)]}'
{"id":"LineageOS%2Fandroid_packages_apps_ManagedProvisioning~318554","triplet_id":"LineageOS%2Fandroid_packages_apps_ManagedProvisioning~lineage-18.1~I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95","project":"LineageOS/android_packages_apps_ManagedProvisioning","branch":"lineage-18.1","topic":"R_asb_2021-11","hashtags":[],"change_id":"I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95","subject":"Fix INTERACT_ACROSS_PROFILES appop issue for shared UIDs","status":"MERGED","created":"2021-11-03 14:16:21.000000000","updated":"2021-11-07 16:58:52.000000000","submitted":"2021-11-07 16:58:52.000000000","submitter":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"318553-R_asb_2021-11","meta_rev_id":"5301f020ab051b819ee3d508ea5313c6db50e3d9","_number":318554,"virtual_id_number":318554,"owner":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"actions":{},"labels":{"Verified":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-1":"Fails"," 0":"No score","+1":"Verified"},"description":"","default_value":0},"Code-Review":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-2":"Do not submit","-1":"I would prefer that you didn\u0027t submit this"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me, approved"},"description":"","default_value":0},"CI":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-1":"Fail"," 0":"No score","+1":"Pass"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{},"pending_reviewers":{},"reviewer_updates":[],"messages":[{"id":"8803e5cfd952a70d8e77c382511fca59e313c47e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"date":"2021-11-03 14:16:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5301f020ab051b819ee3d508ea5313c6db50e3d9","tag":"autogenerated:gerrit:merged","author":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"date":"2021-11-07 16:58:52.000000000","message":"Change has been successfully pushed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5d68ad4753576a15ca9ad52180094dfb727ead4","revisions":{"5ccee481c7f6a56a89a9cdac8c4830a64571f1e6":{"kind":"REWORK","_number":1,"created":"2021-11-03 14:16:21.000000000","uploader":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"ref":"refs/changes/54/318554/1","fetch":{"anonymous http":{"url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning","ref":"refs/changes/54/318554/1","commands":{"Branch":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1 \u0026\u0026 git checkout -b change-318554 FETCH_HEAD","Checkout":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1","Reset To":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/1 \u0026\u0026 git reset --hard FETCH_HEAD"}}},"commit":{"parents":[{"commit":"21aade58eada63d711452052e06a6f196db2644f","subject":"Grant MANAGE_APP_OPS_MODES to Managed Provisioning","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/21aade58eada63d711452052e06a6f196db2644f"}]}],"author":{"name":"kholoud mohamed","email":"kholoudm@google.com","date":"2021-05-07 09:50:28.000000000","tz":60},"committer":{"name":"Kevin F. Haggerty","email":"haggertk@lineageos.org","date":"2021-11-03 12:10:16.000000000","tz":-360},"subject":"Fix INTERACT_ACROSS_PROFILES appop issue for shared UIDs","message":"Fix INTERACT_ACROSS_PROFILES appop issue for shared UIDs\n\nCrossProfileApps#setInteractAcrossProfilesAppOp was previously setting\nthe appop per-package instead of per-UID, causing issues for apps with\nshared UIDs. This is now fixed, however we need to apply the fix in the\nnew OTA update to existing apps that already have the appop set incorrectly.\n\nTest: atest UpdateInteractAcrossProfilesAppOpTaskTest\nTest: manual testing by triggering an OTA update manually\nBug: 183188804\nBug: 183730243\nBug: 195630721\nChange-Id: I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95\nMerged-In: I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95\n(cherry picked from commit 854dad1bc01e40764cec0e278c04c9b33197a80a)\n(cherry picked from commit a356e46bcee7be220e6f0d946dee61986c0ed178)\n","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/5ccee481c7f6a56a89a9cdac8c4830a64571f1e6"}],"resolve_conflicts_web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/5ccee481c7f6a56a89a9cdac8c4830a64571f1e6"}]},"branch":"refs/heads/lineage-18.1"},"e5d68ad4753576a15ca9ad52180094dfb727ead4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-07 16:58:52.000000000","uploader":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"ref":"refs/changes/54/318554/2","fetch":{"anonymous http":{"url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning","ref":"refs/changes/54/318554/2","commands":{"Branch":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2 \u0026\u0026 git checkout -b change-318554 FETCH_HEAD","Checkout":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2","Reset To":"git fetch https://github.com/LineageOS/android_packages_apps_ManagedProvisioning refs/changes/54/318554/2 \u0026\u0026 git reset --hard FETCH_HEAD"}}},"commit":{"parents":[{"commit":"bfdb706c37efdc18a34852d332496d6c17dc3116","subject":"Grant MANAGE_APP_OPS_MODES to Managed Provisioning","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/bfdb706c37efdc18a34852d332496d6c17dc3116"}]}],"author":{"name":"kholoud mohamed","email":"kholoudm@google.com","date":"2021-05-07 09:50:28.000000000","tz":60},"committer":{"name":"Android Build Coastguard Worker","email":"android-build-coastguard-worker@google.com","date":"2021-09-10 03:01:19.000000000","tz":0},"subject":"Fix INTERACT_ACROSS_PROFILES appop issue for shared UIDs","message":"Fix INTERACT_ACROSS_PROFILES appop issue for shared UIDs\n\nCrossProfileApps#setInteractAcrossProfilesAppOp was previously setting\nthe appop per-package instead of per-UID, causing issues for apps with\nshared UIDs. This is now fixed, however we need to apply the fix in the\nnew OTA update to existing apps that already have the appop set incorrectly.\n\nTest: atest UpdateInteractAcrossProfilesAppOpTaskTest\nTest: manual testing by triggering an OTA update manually\nBug: 183188804\nBug: 183730243\nBug: 195630721\nChange-Id: I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95\nMerged-In: I0c826bdc8f0a868cfc7fce3e23c0e124e06b8c95\n(cherry picked from commit 854dad1bc01e40764cec0e278c04c9b33197a80a)\n(cherry picked from commit a356e46bcee7be220e6f0d946dee61986c0ed178)\n","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/e5d68ad4753576a15ca9ad52180094dfb727ead4"}],"resolve_conflicts_web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_packages_apps_ManagedProvisioning/commit/e5d68ad4753576a15ca9ad52180094dfb727ead4"}]},"branch":"refs/heads/lineage-18.1"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
