)]}'
{"id":"LineageOS%2Fandroid_frameworks_base~318547","triplet_id":"LineageOS%2Fandroid_frameworks_base~lineage-18.1~Iced6c2245c0099bc4bcdaceb51f8cce4dbc0a392","project":"LineageOS/android_frameworks_base","branch":"lineage-18.1","topic":"R_asb_2021-11","hashtags":[],"change_id":"Iced6c2245c0099bc4bcdaceb51f8cce4dbc0a392","subject":"TIF: fix issue of using caller-aware methods after clearCallingIdentity()","status":"MERGED","created":"2021-11-03 14:15:50.000000000","updated":"2021-11-07 16:57:49.000000000","submitted":"2021-11-07 16:57:49.000000000","submitter":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"318541-R_asb_2021-11","meta_rev_id":"060b92b55c6ea6a320dc80185bbeab0577f2bcf1","_number":318547,"virtual_id_number":318547,"owner":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"actions":{},"labels":{"Verified":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-1":"Fails"," 0":"No score","+1":"Verified"},"description":"","default_value":0},"Code-Review":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-2":"Do not submit","-1":"I would prefer that you didn\u0027t submit this"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me, approved"},"description":"","default_value":0},"CI":{"all":[{"value":0,"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]}],"values":{"-1":"Fail"," 0":"No score","+1":"Pass"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{},"pending_reviewers":{},"reviewer_updates":[],"messages":[{"id":"5989e0f8b431a38e9a987d5e0daed47ed30bd8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"date":"2021-11-03 14:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"060b92b55c6ea6a320dc80185bbeab0577f2bcf1","tag":"autogenerated:gerrit:merged","author":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"date":"2021-11-07 16:57:49.000000000","message":"Change has been successfully pushed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ee583fd2e4487f1ff4653796ab778219797cfaaa","revisions":{"3712770a8160bbdf636b2e84ee39c4948e2491e5":{"kind":"REWORK","_number":1,"created":"2021-11-03 14:15:50.000000000","uploader":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"ref":"refs/changes/47/318547/1","fetch":{"anonymous http":{"url":"https://github.com/LineageOS/android_frameworks_base","ref":"refs/changes/47/318547/1","commands":{"Branch":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1 \u0026\u0026 git checkout -b change-318547 FETCH_HEAD","Checkout":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1","Reset To":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/1 \u0026\u0026 git reset --hard FETCH_HEAD"}}},"commit":{"parents":[{"commit":"7ab3e347c238a458409e7a1b0d215fbc6aebb3f3","subject":"Backporting the change of ag/15629060 to rvc-dev","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/7ab3e347c238a458409e7a1b0d215fbc6aebb3f3"}]}],"author":{"name":"shubang","email":"shubang@google.com","date":"2021-08-13 22:42:15.000000000","tz":-420},"committer":{"name":"Kevin F. Haggerty","email":"haggertk@lineageos.org","date":"2021-11-03 13:05:26.000000000","tz":-360},"subject":"TIF: fix issue of using caller-aware methods after clearCallingIdentity()","message":"TIF: fix issue of using caller-aware methods after clearCallingIdentity()\n\nBug: 189824175\nTest: atest android.media.tv.cts.TvInputManagerTest\nChange-Id: Iced6c2245c0099bc4bcdaceb51f8cce4dbc0a392\n(cherry picked from commit a2ab2772d70984b1c24d5d4509b9c7a5abea71f9)\n(cherry picked from commit 05363ef193ae231fc7a1fe666de88861e1084b17)\n","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/3712770a8160bbdf636b2e84ee39c4948e2491e5"}],"resolve_conflicts_web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/3712770a8160bbdf636b2e84ee39c4948e2491e5"}]},"branch":"refs/heads/lineage-18.1"},"ee583fd2e4487f1ff4653796ab778219797cfaaa":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-07 16:57:49.000000000","uploader":{"_account_id":15173,"name":"Kevin Haggerty","email":"haggertk@lineageos.org","username":"haggertk","avatars":[{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d32","height":32},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d56","height":56},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d100","height":100},{"url":"https://www.gravatar.com/avatar/258edfac858c1ce5f056ed4ca050a578.jpg?d\u003didenticon\u0026r\u003dpg\u0026s\u003d120","height":120}]},"ref":"refs/changes/47/318547/2","fetch":{"anonymous http":{"url":"https://github.com/LineageOS/android_frameworks_base","ref":"refs/changes/47/318547/2","commands":{"Branch":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2 \u0026\u0026 git checkout -b change-318547 FETCH_HEAD","Checkout":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2","Reset To":"git fetch https://github.com/LineageOS/android_frameworks_base refs/changes/47/318547/2 \u0026\u0026 git reset --hard FETCH_HEAD"}}},"commit":{"parents":[{"commit":"42f9dee6ca9affc8481a51521388ebf0ec623d6b","subject":"Backporting the change of ag/15629060 to rvc-dev","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/42f9dee6ca9affc8481a51521388ebf0ec623d6b"}]}],"author":{"name":"shubang","email":"shubang@google.com","date":"2021-08-13 22:42:15.000000000","tz":-420},"committer":{"name":"Anis Assi","email":"anisassi@google.com","date":"2021-10-01 20:01:01.000000000","tz":0},"subject":"TIF: fix issue of using caller-aware methods after clearCallingIdentity()","message":"TIF: fix issue of using caller-aware methods after clearCallingIdentity()\n\nBug: 189824175\nTest: atest android.media.tv.cts.TvInputManagerTest\nChange-Id: Iced6c2245c0099bc4bcdaceb51f8cce4dbc0a392\n(cherry picked from commit a2ab2772d70984b1c24d5d4509b9c7a5abea71f9)\n(cherry picked from commit 05363ef193ae231fc7a1fe666de88861e1084b17)\n","web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/ee583fd2e4487f1ff4653796ab778219797cfaaa"}],"resolve_conflicts_web_links":[{"name":"GitHub","tooltip":"Open in GitWeb","url":"https://github.com/LineageOS/android_frameworks_base/commit/ee583fd2e4487f1ff4653796ab778219797cfaaa"}]},"branch":"refs/heads/lineage-18.1"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
